Skip to content

paintの翻訳として使われているペイントを描画へと統一しました - #816

Open
dowdiness wants to merge 4 commits into
reactjs:mainfrom
dowdiness:paint
Open

dowdiness wants to merge 4 commits into
reactjs:mainfrom
dowdiness:paint

Conversation

@dowdiness

Copy link
Copy Markdown

ドキュメント内でbrowser renderingの意味で使われているペイントを全て描画に変更しました。このPRによりペイントという単語はドキュメント内に出現しなくなります。

renderとpaintはこのドキュメントにおいて意識的に使い分けられている重要な用語だと思います。しかし日本語版では ドキュメント全体を通して「ペイント」と呼ぶことにします という記述があるにも関わらず、実際には描画という用語が全体として使われていてペイントは出てこなくなります。

過去の議論を見る限り、renderの訳はレンダー、paintの訳は描画に決まっていそうなので、browser renderingとしての意味で使われているpaintは全て描画に統一した方が分かりやすくなるはずです。

wikiの訳語の統一にもpaintの項目を追加したほうが良いかもしれません。

@vercel

vercel Bot commented Dec 25, 2024 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
legacy-ja-reactjs-org Ignored Ignored Preview Oct 5, 2026 8:39am UTC

Request Review

@smikitky smikitky left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

ありがとうございます。

訳語については良いのですが、改行のつけかたがおかしくなっていますので修正をお願いします。

翻訳スタイルガイド

  • 原文と改行のしかた、空行の入れかたを厳密に一致させる。GitHub の差分表示画面で見て行番号レベルで左右が対応しているようにする。これは将来原文が更新された場合に Git が修正箇所を正しく対応づけるために重要。

@smikitky

Copy link
Copy Markdown
Member

ただ、もしかしたら「描画」を「ペイント」に統一したほうがむしろ分かりやすくなるのではないかという気もしています…時間があるときに検討させてください

@dowdiness

Copy link
Copy Markdown
Author

理由があるのを知らず、不自然に改行してない部分があるなと思い適当にやってしまいました。
翻訳スタイルガイド通りに原文と改行のしかた、空行の入れかたを厳密に一致させるよう変更しました。

@dowdiness

Copy link
Copy Markdown
Author

修正のPRを送っておいてなんですが、実は私も描画ではなくペイントにした方が分かりやすいんじゃないかなとは感じています。

レンダリングという言葉が映像やCG、ブラウザの歴史もあり、日本語の中で英語とは少し違う意味を持ち始めているのもあって、描画=レンダリングと脳内で無意識に変換する人が多いように思います。

実際にreactの日本語情報を検索してみると描画がレンダリングと同義語で使われているものが多くあります。

そういう日本語の資料を今まで読んだきた人たちがこのドキュメントで描画という単語に出会った時に勘違いしないように、原文に寄せたカタカナのペイントにしたほうが明確な意図があってこの言葉を選んでいるのが分かりやすくなり混乱が起きにくい気がします。(あくまで個人的な意見です!年末ですのでまた時間のある時にゆっくり検討してください)

@github-actions

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

This PR introduced no changes to the JavaScript bundle! 🙌

@smikitky

smikitky commented Oct 5, 2026 •

Copy link
Copy Markdown
Member

長らく放置しており申し訳ありません。

こちらですが、本来の新ドキュメント作者 (Dan) の意図としては、ここで宣言されている通り、paint に厳密に統一するはずでした。しかし後に英語ドキュメントをいろんな人が編集する過程でまたその辺の一貫性が崩れ、browser paint のことを深く考えずに render と書いてしまう人がまた増えてきてしまっている、という状況です。

今後は文脈上でブラウザエンジンのレンダリングを指しているrender/paintに関しては、以下のようにしたいと思います。

  • 原文が(誤って)render の場合:意味上は「ペイント」に統一してしまってもいいのだが、用語ガイドラインの通り今は「描画」でお茶を濁す
  • 原文が(ちゃんと)paint の場合:「描画」で翻訳されてしまっているところも含め、「ペイント」に統一する

何もかもを「描画」に統一するのも一案ですが、日本語ドキュメントで一旦「これはペイントと呼びます」と宣言していることをひっくり返すことになりますし、個人的にも元の Dan の思想を尊重してカタカナのままの方が読みやすいことが多いかなと思います。

この方針で、不適切な「描画」を「ペイント」に統一しました。差分 610c659 は原文が paint になっているものを「描画」→「ペイント」にするものだけのはずです。

@honey32 ご確認お願いします。

This branch was successfully deployed

1 active deployment
Preview – ja-react-dev — c05039de Deployed Oct 5, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants